Skip to content

chore(deps): update dependency eslint-plugin-import to v2.32.0 - #258

Open
renovate[bot] wants to merge 1 commit into
mainfrom
renovate/eslint-plugin-import-2.x-lockfile
Open

renovate[bot] wants to merge 1 commit into
mainfrom
renovate/eslint-plugin-import-2.x-lockfile

Conversation

@renovate

@renovate renovate Bot commented Sep 22, 2025

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Change Age Confidence
eslint-plugin-import 2.31.02.32.0 age confidence

Release Notes

import-js/eslint-plugin-import (eslint-plugin-import)

v2.32.0

Compare Source

Added
Fixed
Changed

Configuration

📅 Schedule: (UTC)

  • Branch creation
    • At any time (no schedule defined)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR was generated by Mend Renovate. View the repository job log.

@nx-cloud

nx-cloud Bot commented Sep 22, 2025

Copy link
Copy Markdown

View your CI Pipeline Execution ↗ for commit 4a02efd

Command Status Duration Result
nx affected -t lint test build ✅ Succeeded 47s View ↗

💡 Verify your cache is correct by running tasks in a sandbox. Read docs ↗


☁️ Nx Cloud last updated this comment at 2026-09-15 15:44:35 UTC

@renovate
renovate Bot force-pushed the renovate/eslint-plugin-import-2.x-lockfile branch from c761601 to 26bed0f Compare September 25, 2025 20:07
@renovate
renovate Bot force-pushed the renovate/eslint-plugin-import-2.x-lockfile branch from 26bed0f to dde7223 Compare October 22, 2025 01:07
@sonarqubecloud

Copy link
Copy Markdown

@renovate
renovate Bot force-pushed the renovate/eslint-plugin-import-2.x-lockfile branch from dde7223 to 01a6975 Compare December 31, 2025 16:15
@sonarqubecloud

Copy link
Copy Markdown

@renovate
renovate Bot force-pushed the renovate/eslint-plugin-import-2.x-lockfile branch from 01a6975 to 8754de5 Compare February 12, 2026 14:58
@renovate
renovate Bot force-pushed the renovate/eslint-plugin-import-2.x-lockfile branch from 8754de5 to 4d47ea5 Compare March 5, 2026 19:40
@renovate
renovate Bot force-pushed the renovate/eslint-plugin-import-2.x-lockfile branch 2 times, most recently from 205467c to a8af48d Compare April 8, 2026 14:38
@renovate
renovate Bot force-pushed the renovate/eslint-plugin-import-2.x-lockfile branch from a8af48d to 172a3fd Compare April 29, 2026 16:17
@sonarqubecloud

Copy link
Copy Markdown

@renovate
renovate Bot force-pushed the renovate/eslint-plugin-import-2.x-lockfile branch from 172a3fd to 21d9512 Compare May 30, 2026 23:38
@sonarqubecloud

Copy link
Copy Markdown

@renovate
renovate Bot force-pushed the renovate/eslint-plugin-import-2.x-lockfile branch from 21d9512 to a93e8d6 Compare July 12, 2026 11:53
@renovate
renovate Bot force-pushed the renovate/eslint-plugin-import-2.x-lockfile branch from a93e8d6 to c7704b7 Compare July 30, 2026 17:16
@sonarqubecloud

Copy link
Copy Markdown

@renovate
renovate Bot force-pushed the renovate/eslint-plugin-import-2.x-lockfile branch from c7704b7 to 358214d Compare September 10, 2026 16:35
@renovate
renovate Bot force-pushed the renovate/eslint-plugin-import-2.x-lockfile branch from 358214d to 4a02efd Compare September 15, 2026 15:42
@sonarqubecloud

Copy link
Copy Markdown

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants